Repository navigation
[experimental] Add SINDI IP U12 + E5M7 storage - #1864
Open
alexanderguzhva wants to merge 2 commits into
Open
alexanderguzhva wants to merge 2 commits into
alexanderguzhva wants to merge 2 commits into
Conversation
Signed-off-by: Alexandr Guzhva <alexanderguzhva@gmail.com>
Collaborator
|
[APPROVALNOTIFIER] This PR is APPROVED This pull-request has been approved by: alexanderguzhva The full list of commands accepted by this bot can be found here. The pull request process is described here DetailsNeeds approval from an approver in each of these files:
Approvers can indicate their approval by writing |
|
@alexanderguzhva 🔍 Important: PR Classification Needed! For efficient project management and a seamless review process, it's essential to classify your PR correctly. Here's how:
For any PR outside the kind/improvement category, ensure you link to the associated issue using the format: “issue: #”. Thanks for your efforts and contribution to the community!. |
alexanderguzhva
force-pushed
the
sindi_ip_u12_e5m7
branch
from
September 30, 2026 15:27
418b139 to
b5d7af0
Compare
Collaborator
Author
|
/kind improvement |
Signed-off-by: Alexandr Guzhva <alexanderguzhva@gmail.com>
alexanderguzhva
force-pushed
the
sindi_ip_u12_e5m7
branch
from
September 30, 2026 17:45
b5d7af0 to
81bb852
Compare
This file contains hidden or bidirectional Unicode text that may be interpreted or compiled differently than what appears below. To review, open the file in an editor that reveals hidden Unicode characters.
Learn more about bidirectional Unicode characters
Sign up for free
to join this conversation on GitHub.
Already have an account?
Sign in to comment
Add this suggestion to a batch that can be applied as a single commit.This suggestion is invalid because no changes were made to the code.Suggestions cannot be applied while the pull request is closed.Suggestions cannot be applied while viewing a subset of changes.Only one suggestion per line can be applied in a batch.Add this suggestion to a batch that can be applied as a single commit.Applying suggestions on deleted lines is not supported.You must change the existing code in this line in order to create a valid suggestion.Outdated suggestions cannot be applied.This suggestion has been applied or marked resolved.Suggestions cannot be applied from pending reviews.Suggestions cannot be applied on multi-line comments.Suggestions cannot be applied while the pull request is queued to merge.Suggestion cannot be applied right now. Please check back later.
This is an experimental PR, which tests U12 instead of U16 and E5M7 instead of FP16 for SINDI IP. It trades the QPS & recall for the smaller database size (~25% less for MSMARCO / SPLADE).
This PR is based on top of #1856, so the refinement is also enabled.
The refinement pareto frontier for MSMARCO/SPLADE is the following for the G4 8c32g machine (results are directly comparable with the one from the description of #1856):