Skip to content

feat: remove xo - #441

Merged
pataruco merged 1 commit into
mainfrom
remove-xo
Sep 29, 2026
Merged

pataruco merged 1 commit into
mainfrom
remove-xo

Conversation

@pataruco

Copy link
Copy Markdown
Contributor

No description provided.

@pataruco pataruco self-assigned this Sep 29, 2026
@nx-cloud

nx-cloud Bot commented Sep 29, 2026

Copy link
Copy Markdown

View your CI Pipeline Execution ↗ for commit 149f82f

Command Status Duration Result
nx affected run-many -t test --base=520ca706211... ✅ Succeeded 5s View ↗
nx affected run-many -t lint format --base=520c... ✅ Succeeded <1s View ↗

💡 Verify your cache is correct by running tasks in a sandbox. Read docs ↗


☁️ Nx Cloud last updated this comment at 2026-09-29 19:20:15 UTC

@nx-cloud

nx-cloud Bot commented Sep 29, 2026

Copy link
Copy Markdown

View your CI Pipeline Execution ↗ for commit 149f82f

Command Status Duration Result
nx affected run-many -t test --base=520ca706211... ✅ Succeeded 5s View ↗
nx affected run-many -t lint format --base=520c... ✅ Succeeded <1s View ↗

💡 Verify your cache is correct by running tasks in a sandbox. Read docs ↗


☁️ Nx Cloud last updated this comment at 2026-09-29 19:20:16 UTC

@pataruco
pataruco merged commit 0028f56 into main Sep 29, 2026
10 checks passed
@pataruco
pataruco deleted the remove-xo branch September 29, 2026 19:21
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

None yet

Projects

None yet

Development

Successfully merging this pull request may close these issues.

1 participant