Skip to content

Changeset bump and fix docs readme - #3778

Merged
portuu3 merged 1 commit into
developfrom
release20260211
Feb 11, 2026
Merged

portuu3 merged 1 commit into
developfrom
release20260211

Conversation

@portuu3

@portuu3 portuu3 commented Feb 11, 2026

Copy link
Copy Markdown
Collaborator

Issue tracking

Changeset bump for #3749

Context behind the change

  • Changeset bump
  • Fix docs readme

How has this been tested?

NA

Release plan

NA

@portuu3 portuu3 self-assigned this Feb 11, 2026
@vercel

vercel Bot commented Feb 11, 2026

Copy link
Copy Markdown

The latest updates on your projects. Learn more about Vercel for GitHub.

Project Deployment Actions Updated (UTC)
human-app Ready Ready Preview, Comment Feb 11, 2026 6:56am
human-dashboard-frontend Ready Ready Preview, Comment Feb 11, 2026 6:56am
staking-dashboard Ready Ready Preview, Comment Feb 11, 2026 6:56am
2 Skipped Deployments
Project Deployment Actions Updated (UTC)
faucet-frontend Ignored Ignored Feb 11, 2026 6:56am
faucet-server Ignored Ignored Feb 11, 2026 6:56am

Request Review

@portuu3
portuu3 merged commit a6df342 into develop Feb 11, 2026
22 checks passed
@github-project-automation github-project-automation Bot moved this from In Review to Done in Core tech - 2026 (H1) Feb 11, 2026
@portuu3
portuu3 deleted the release20260211 branch February 11, 2026 09:11
@portuu3 portuu3 mentioned this pull request Feb 11, 2026
30 tasks done

This branch was successfully deployed

3 active deployments
Preview – human-app — de706534 Deployed Feb 11, 2026 by vercel[bot]
Preview – human-dashboard-frontend — de706534 Deployed Feb 11, 2026 by vercel[bot]
Preview – staking-dashboard — de706534 Deployed Feb 11, 2026 by vercel[bot]
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

None yet

Projects

Status: Done

Development

Successfully merging this pull request may close these issues.

2 participants