Skip to content

feat(core): add workspaces command - #22

Merged
aid7n merged 1 commit into
mainfrom
feat/workspaces-cmd
Jul 28, 2026
Merged

aid7n merged 1 commit into
mainfrom
feat/workspaces-cmd

Conversation

@aid7n

@aid7n aid7n commented Jul 28, 2026

Copy link
Copy Markdown
Owner

No description provided.

@aid7n aid7n self-assigned this Jul 28, 2026
@changeset-bot

changeset-bot Bot commented Jul 28, 2026

Copy link
Copy Markdown

🦋 Changeset detected

Latest commit: 489c6c2

The changes in this PR will be included in the next version bump.

This PR includes changesets to release 9 packages
Name Type
@7x/jb-core Minor
@7x/jb Patch
@7x/jb-config Minor
@7x/jb-darwin-arm64 Minor
@7x/jb-darwin-x64 Minor
@7x/jb-linux-arm64 Minor
@7x/jb-linux-x64 Minor
@7x/jb-win32-arm64 Minor
@7x/jb-win32-x64 Minor

Not sure what this means? Click here to learn what changesets are.

Click here if you're a maintainer who wants to add another changeset to this PR

@aid7n
aid7n merged commit b982d17 into main Jul 28, 2026
1 check passed
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

None yet

Projects

None yet

Development

Successfully merging this pull request may close these issues.

1 participant